Conversation
13 tasks
taranarmo
force-pushed
the
aiohttp-swagger-fix-tests
branch
from
May 5, 2025 12:53
b82211e to
e52f335
Compare
Contributor
Author
|
@jopejoe1 I'm sorry to bother but seems like this fixes tests for I rebased the branch on the last commit before deleting tests and it built successfully. |
Member
|
You probly want to do a revert commit before your commit so that there is no longer a merge conflict |
jopejoe1
reviewed
May 5, 2025
taranarmo
force-pushed
the
aiohttp-swagger-fix-tests
branch
from
May 6, 2025 07:45
e52f335 to
2ebeeda
Compare
Contributor
Author
|
Thank you for your help, I created a PR upstream to fix the tests and added those commits here as patches, now it builds fine. |
taranarmo
force-pushed
the
aiohttp-swagger-fix-tests
branch
from
May 6, 2025 08:08
2ebeeda to
35284b3
Compare
This reverts commit ba4935b. Patches to fix tests are added in the previous commit e1f478b738d5390cc553f851deb252f1e3cad2bd.
mkg20001
force-pushed
the
aiohttp-swagger-fix-tests
branch
from
August 29, 2025 18:13
35284b3 to
b2e59b0
Compare
Member
|
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
During attempt to build
apache-airflow 2.10.5I faced failed tests onaiohttp-swaggerpackage build due to unsetasyncio_default_fixture_loop_scopeoption which is now deprecatedpytest-dev/pytest-asyncio#924
This PR sets it with environmental variable.
This is also included in #396544.
Things done
nix.conf? (See Nix manual)sandbox = relaxedsandbox = truenix-shell -p nixpkgs-review --run "nixpkgs-review rev HEAD". Note: all changes have to be committed, also see nixpkgs-review usage./result/bin/)Add a 👍 reaction to pull requests you find important.