Skip to content

chore(deps): bump azure/login from 3.0.0 to 3.0.1 - #3969

Merged
jgravois merged 1 commit into
mainfrom
dependabot/github_actions/azure/login-3.0.1
Aug 12, 2026
Merged

chore(deps): bump azure/login from 3.0.0 to 3.0.1#3969
jgravois merged 1 commit into
mainfrom
dependabot/github_actions/azure/login-3.0.1

Conversation

@dependabot

@dependabot dependabot Bot commented on behalf of github Aug 7, 2026

Copy link
Copy Markdown
Contributor

Bumps azure/login from 3.0.0 to 3.0.1.

Release notes

Sourced from azure/login's releases.

Azure Login Action v3.0.1

What's Changed

Full Changelog: https://github.com/Azure/login/commits/v3.0.1

Commits

@dependabot dependabot Bot added the dependencies [auto] Pull requests that update 3rd party software libraries and requirements label Aug 7, 2026
@dependabot
dependabot Bot requested a review from a team as a code owner August 7, 2026 21:42
@github-actions

github-actions Bot commented Aug 7, 2026

Copy link
Copy Markdown
Contributor

Coverage report

This PR does not seem to contain any modification to coverable code.

@thekaveman

Copy link
Copy Markdown
Member

@dependabot rebase

Bumps [azure/login](https://github.com/azure/login) from 3.0.0 to 3.0.1.
- [Release notes](https://github.com/azure/login/releases)
- [Commits](Azure/login@532459e...f5d393a)

---
updated-dependencies:
- dependency-name: azure/login
  dependency-version: 3.0.1
  dependency-type: direct:production
  update-type: version-update:semver-patch
...

Signed-off-by: dependabot[bot] <support@github.com>
@dependabot
dependabot Bot force-pushed the dependabot/github_actions/azure/login-3.0.1 branch from 4ec6ae6 to 66468a5 Compare August 10, 2026 23:35
@jgravois

Copy link
Copy Markdown
Member

I'd like to look into updating azurerm too. I think we're ~20 versions behind...

@thekaveman

Copy link
Copy Markdown
Member

I'd like to look into updating azurerm too. I think we're ~20 versions behind...

Sure, probably a different PR yeah?

Any idea why this plan run is failing though? The error message indicates we don't have the step configured correctly, but of course that is wrong 🤔

@jgravois

Copy link
Copy Markdown
Member

different PR for sure.

this error message is a mystery to me. we're supplying the appropriate input params and the patch release notes don't provide any clues 🤔

@jgravois

Copy link
Copy Markdown
Member

ah, of course. dependabot doesn't have access to repository secrets. i'll create duplicate values that it can access and rerun the failing job.

@thekaveman

Copy link
Copy Markdown
Member

Ah yes 🤦

@github-actions

Copy link
Copy Markdown
Contributor
CHANGE RESOURCE
update (2) module.application.azurerm_container_app.pgadmin
module.application.azurerm_container_app.web

@jgravois jgravois left a comment

Copy link
Copy Markdown
Member

Choose a reason for hiding this comment

The reason will be displayed to describe this comment to others. Learn more.

i think this one is safe to merge but i'm assigning @lalver1 too so that he can review the terraform plan output and confirm that its safe to deploy the changes it describes to the dev environment.

now that he's load-testing test i think so, but better safe than sorry.

@jgravois
jgravois requested a review from lalver1 August 12, 2026 18:50
@jgravois

jgravois commented Aug 12, 2026

Copy link
Copy Markdown
Member

i'm rescinding @lalver1's review here because merging #3980 deployed the plan i mentioned above to dev.

all good though! 😇

@jgravois
jgravois removed the request for review from lalver1 August 12, 2026 21:01
@jgravois
jgravois merged commit 9b0dd4c into main Aug 12, 2026
9 of 11 checks passed
@jgravois
jgravois deleted the dependabot/github_actions/azure/login-3.0.1 branch August 12, 2026 21:02
@thekaveman

Copy link
Copy Markdown
Member

merging #3980 deployed the plan i mentioned above to dev.

Interesting. I think this is unexpected at least so far as my PR was concerned -- we didn't get any plan summary comment, so I didn't realize anything was changing (beyond the expected SHA).

Should we change/remove the path filters for plan?

@jgravois

jgravois commented Aug 12, 2026

Copy link
Copy Markdown
Member

Should we change/remove the path filters for plan?

personally i don't see what happened here as a bug. the drift on dev was the result of @lalver1's fiddling. not anything to do with your PR.

  1. he temporarily allow-listed his own IP (to connect to pgAdmin)
  2. he temporarily disabled reCaptcha (by mucking with the secrets loaded in the container app)

@thekaveman

Copy link
Copy Markdown
Member

Agreed not a bug per-se, but it highlights an issue that plan and the summary is supposed to solve for us: We should know for a fact what infra changes will be made when merging any random PR.

It is always possible DevSecOps twiddles some nobs and those changes don't make their way down to our Terraform, or something else (more nefarious maybe??)

I still think over-communication here is better than an apply that (un)does things unexpectedly.

@jgravois

Copy link
Copy Markdown
Member

I still think over-communication here is better than an apply that (un)does things unexpectedly.

i can get behind that, but its worth noting that even time itself (between the last commit pushed and a PRs eventual merge) opens us up to the potential of apply not adhering to the plan.

another cool option might be to consider surfacing terraform apply summaries more prominently. (especially when they do more than rev the Container app's SHA)

https://github.com/cal-itp/benefits/actions/runs/31639968999

@thekaveman

Copy link
Copy Markdown
Member

even time itself (between the last commit pushed and a PRs eventual merge) opens us up to the potential of apply not adhering to the plan.

Fair enough. This all raises a related question for me, that isn't a new issue, but maybe still worth discussing:

How do we know what changes will apply to test or prod without just pushing a tag and watching the Actions output?

I think we can leave it here for now, and maybe revisit all of this in a future Workshop convo.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

dependencies [auto] Pull requests that update 3rd party software libraries and requirements

Projects

None yet

Development

Successfully merging this pull request may close these issues.

3 participants