Skip to content

feat(experimental): getAssets (ERC-7811) - #3843

Merged
jxom merged 5 commits into
mainfrom
jxom/get-assets
Aug 7, 2025
Merged

jxom merged 5 commits into
mainfrom
jxom/get-assets

Conversation

@jxom

@jxom jxom commented Aug 5, 2025

Copy link
Copy Markdown
Member

No description provided.

@changeset-bot

changeset-bot Bot commented Aug 5, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: eb79baf

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
viem Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@jxom jxom changed the title feat: getAssets feat(experimental): getAssets Aug 5, 2025
@jxom jxom changed the title feat(experimental): getAssets feat(experimental): getAssets (ERC-7811) Aug 5, 2025
@jxom
jxom marked this pull request as ready for review August 5, 2025 19:34
@github-actions

github-actions Bot commented Aug 5, 2025

Copy link
Copy Markdown
Contributor

size-limit report 📦

Path Size Loading time (3g) Running time (snapdragon) Total time
import * from 'viem' (esm) 71.1 KB (0%) 1.5 s (0%) 991 ms (+43.85% 🔺) 2.5 s
const viem = require('viem') (cjs) 87.98 KB (0%) 1.8 s (0%) 2.9 s (+22.61% 🔺) 4.7 s
import { createClient, http } from 'viem' 4.63 KB (0%) 93 ms (0%) 71 ms (-10.02% 🔽) 163 ms
import * from 'viem/account-abstraction' 53.66 KB (0%) 1.1 s (0%) 1.1 s (+15.21% 🔺) 2.2 s
import * from 'viem/accounts' 82.5 KB (0%) 1.7 s (0%) 757 ms (+107.43% 🔺) 2.5 s
import { privateKeyToAccount } from 'viem/accounts' 21.04 KB (0%) 421 ms (0%) 361 ms (+76.21% 🔺) 782 ms
import { mnemonicToAccount } from 'viem/accounts' 27.35 KB (0%) 547 ms (0%) 385 ms (-53.41% 🔽) 932 ms
import * from 'viem/actions' 55.74 KB (0%) 1.2 s (0%) 1.2 s (+42.87% 🔺) 2.3 s
import { getBlockNumber } from 'viem/actions' 318 B (0%) 10 ms (0%) 59 ms (+145.28% 🔺) 69 ms
import * from 'viem/chains' 48.22 KB (0%) 965 ms (0%) 1.2 s (+46.45% 🔺) 2.2 s
import { mainnet } from 'viem/chains' 325 B (0%) 10 ms (0%) 26 ms (+658.93% 🔺) 36 ms
import * from 'viem/chains/utils' 1.08 KB (0%) 22 ms (0%) 22 ms (+71.05% 🔺) 44 ms
import * from 'viem/ens' 47.85 KB (0%) 957 ms (0%) 657 ms (+29.1% 🔺) 1.7 s
import { getEnsAvatar } from 'viem/ens' 24.04 KB (0%) 481 ms (0%) 459 ms (+54.98% 🔺) 940 ms
import * from 'viem/siwe' 33.13 KB (0%) 663 ms (0%) 635 ms (+12.22% 🔺) 1.3 s
import { verifySiweMessage } from 'viem/siwe' 32.02 KB (0%) 641 ms (0%) 612 ms (-20.89% 🔽) 1.3 s

@jxom
jxom force-pushed the jxom/get-assets branch from 3adfc7d to eb79baf Compare August 7, 2025 08:14
@jxom
jxom merged commit 15352db into main Aug 7, 2025
2 checks passed
@jxom
jxom deleted the jxom/get-assets branch August 7, 2025 08:18
@github-actions github-actions Bot mentioned this pull request Aug 7, 2025
@codecov

codecov Bot commented Aug 7, 2025

Copy link
Copy Markdown

Codecov Report

❌ Patch coverage is 99.45355% with 2 lines in your changes missing coverage. Please review.
✅ Project coverage is 99.59%. Comparing base (4a5249a) to head (9cdd69f).
⚠️ Report is 5 commits behind head on main.

Files with missing lines Patch % Lines
src/experimental/erc7811/actions/getAssets.ts 99.19% 2 Missing ⚠️
Additional details and impacted files
@@            Coverage Diff             @@
##             main    #3843      +/-   ##
==========================================
+ Coverage   97.35%   99.59%   +2.24%     
==========================================
  Files        1034     1041       +7     
  Lines       80477    81053     +576     
  Branches     3094     3530     +436     
==========================================
+ Hits        78350    80727    +2377     
+ Misses       2090      304    -1786     
+ Partials       37       22      -15     

☔ View full report in Codecov by Sentry.
📢 Have feedback on the report? Share it here.

🚀 New features to boost your workflow:
  • ❄️ Test Analytics: Detect flaky tests, report on failures, and find test suite problems.
  • 📦 JS Bundle Analysis: Save yourself from yourself by tracking and limiting bundle sizes in JS merges.

Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

1 participant