Add Operator Fee for OP Stack Chains - #3950
Merged
Merged
Conversation
🦋 Changeset detectedLatest commit: 81c4ba3 The changes in this PR will be included in the next version bump. This PR includes changesets to release 1 package
Not sure what this means? Click here to learn what changesets are. Click here if you're a maintainer who wants to add another changeset to this PR |
tmm
force-pushed
the
main
branch
3 times, most recently
from
September 19, 2025 19:48
129b947 to
9ad3509
Compare
jxom
force-pushed
the
main
branch
2 times, most recently
from
October 7, 2025 01:53
d78cf00 to
b31cf2a
Compare
commit: |
Contributor
Author
|
@jxom any thoughts on this one? |
jxom
reviewed
Nov 27, 2025
jxom
reviewed
Nov 27, 2025
jxom
reviewed
Nov 27, 2025
jxom
requested changes
Nov 27, 2025
jxom
left a comment
Member
There was a problem hiding this comment.
Please add estimateOperatorFee to op-stack/decorators/publicL2.ts
jxom
reviewed
Nov 27, 2025
Co-authored-by: jxom <7336481+jxom@users.noreply.github.com>
Co-authored-by: jxom <7336481+jxom@users.noreply.github.com>
Co-authored-by: jxom <7336481+jxom@users.noreply.github.com>
Co-authored-by: jxom <7336481+jxom@users.noreply.github.com>
jxom
approved these changes
Dec 1, 2025
Merged
Contributor
Author
|
🚀 thanks! go viem |
This file contains hidden or bidirectional Unicode text that may be interpreted or compiled differently than what appears below. To review, open the file in an editor that reveals hidden Unicode characters.
Learn more about bidirectional Unicode characters
Sign up for free
to join this conversation on GitHub.
Already have an account?
Sign in to comment
Add this suggestion to a batch that can be applied as a single commit.This suggestion is invalid because no changes were made to the code.Suggestions cannot be applied while the pull request is closed.Suggestions cannot be applied while viewing a subset of changes.Only one suggestion per line can be applied in a batch.Add this suggestion to a batch that can be applied as a single commit.Applying suggestions on deleted lines is not supported.You must change the existing code in this line in order to create a valid suggestion.Outdated suggestions cannot be applied.This suggestion has been applied or marked resolved.Suggestions cannot be applied from pending reviews.Suggestions cannot be applied on multi-line comments.Suggestions cannot be applied while the pull request is queued to merge.Suggestion cannot be applied right now. Please check back later.
Adding Operator Fee to the OP Stack actions. The Operator Fee is supported on OP Stack since the Isthmus Upgrade:
https://docs.optimism.io/stack/transactions/fees#operator-fee