Skip to content

Add blockTime property to Scroll chain definition - #4033

Merged
jxom merged 2 commits into
wevm:mainfrom
boredland:patch-7
Oct 16, 2025
Merged

jxom merged 2 commits into
wevm:mainfrom
boredland:patch-7

Conversation

@boredland

Copy link
Copy Markdown
Contributor

@changeset-bot

changeset-bot Bot commented Oct 16, 2025

Copy link
Copy Markdown

🦋 Changeset detected

Latest commit: 4ee193c

The changes in this PR will be included in the next version bump.

This PR includes changesets to release 1 package
Name Type
viem Patch

Not sure what this means? Click here to learn what changesets are.

Click here if you're a maintainer who wants to add another changeset to this PR

@pkg-pr-new

pkg-pr-new Bot commented Oct 16, 2025

Copy link
Copy Markdown

Open in StackBlitz

npm i https://pkg.pr.new/viem@4033

commit: 38958b5

@jxom
jxom merged commit 5601fc1 into wevm:main Oct 16, 2025
11 of 25 checks passed
@github-actions github-actions Bot mentioned this pull request Oct 16, 2025
@tmm tmm mentioned this pull request Oct 23, 2025
Sign up for free to join this conversation on GitHub. Already have an account? Sign in to comment

Labels

None yet

Projects

None yet

Development

Successfully merging this pull request may close these issues.

2 participants